Update docker.io/golang Docker tag to v1.27 - #1915
renovate-bot wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: renovate-bot The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
|
This pull request has been automatically marked as stale because it has not had recent activity. It will be closed in 30 days if no further activity occurs. Thank you for your contributions. |
|
This pull request has been automatically marked as stale because it has not had recent activity. It will be closed in 30 days if no further activity occurs. Thank you for your contributions. |
|
This pull request has been automatically marked as stale because it has not had recent activity. It will be closed in 30 days if no further activity occurs. Thank you for your contributions. |
Signed-off-by: Mend Renovate <bot@renovateapp.com>
ac39cc8 to
dc50da2
Compare
|
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #1915 +/- ##
=======================================
- Coverage 68.8% 68.7% -0.1%
=======================================
Files 58 58
Lines 6076 6076
=======================================
- Hits 4184 4178 -6
- Misses 1582 1585 +3
- Partials 310 313 +3 🚀 New features to boost your workflow:
|



This PR contains the following updates:
1.25→1.27Configuration
📅 Schedule: (in timezone America/New_York)
* 3-6 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.