Update github.com/openshift/controller-runtime-common digest to e1187ec - #1953
renovate-bot wants to merge 1 commit into
Conversation
ℹ️ Artifact update noticeFile name: go.modIn order to perform the update(s) described in the table above, Renovate ran the
Details:
|
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: renovate-bot The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
41abdac to
99e0c61
Compare
99e0c61 to
c864f07
Compare
c864f07 to
4083e01
Compare
|
4083e01 to
190703a
Compare
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #1953 +/- ##
=======================================
- Coverage 68.7% 68.6% -0.1%
=======================================
Files 58 58
Lines 6056 6056
=======================================
- Hits 4164 4158 -6
- Misses 1581 1584 +3
- Partials 311 314 +3 🚀 New features to boost your workflow:
|
190703a to
6c50c3f
Compare
6c50c3f to
83d237e
Compare
|
0ed91a6 to
9251fa3
Compare
9251fa3 to
539f5ea
Compare
539f5ea to
aa08d65
Compare
|
|
This pull request has been automatically marked as stale because it has not had recent activity. It will be closed in 30 days if no further activity occurs. Thank you for your contributions. |
|
aa08d65 to
c0e192e
Compare
Signed-off-by: Mend Renovate <bot@renovateapp.com>
c0e192e to
d3bf1da
Compare
|



This PR contains the following updates:
5db94f6→e1187ecConfiguration
📅 Schedule: (in timezone America/New_York)
* 3-6 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.