Skip to content

Update k8s.io/utils digest to cf1189d - #2060

Open
renovate-bot wants to merge 1 commit into
backube:mainfrom
renovate-bot:renovate/k8s.io-utils-digest
Open

renovate-bot wants to merge 1 commit into
backube:mainfrom
renovate-bot:renovate/k8s.io-utils-digest

Conversation

@renovate-bot

@renovate-bot renovate-bot commented Jul 8, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
k8s.io/utils require digest be93311cf1189d

Configuration

📅 Schedule: (in timezone America/New_York)

  • Branch creation
    • Between 03:00 AM and 06:59 AM, Monday through Friday (* 3-6 * * 1-5)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@forking-renovate forking-renovate Bot added the dependencies Pull requests that update a dependency file label Jul 8, 2026
@openshift-ci

openshift-ci Bot commented Jul 8, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: renovate-bot
Once this PR has been reviewed and has the lgtm label, please assign johnstrunk for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the size/XS label Jul 8, 2026
@codecov

codecov Bot commented Jul 8, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 68.8%. Comparing base (7ec5ae9) to head (63a7b76).

Additional details and impacted files
@@          Coverage Diff          @@
##            main   #2060   +/-   ##
=====================================
  Coverage   68.7%   68.8%           
=====================================
  Files         58      58           
  Lines       6076    6076           
=====================================
+ Hits        4178    4184    +6     
+ Misses      1585    1582    -3     
+ Partials     313     310    -3     

see 1 file with indirect coverage changes

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@renovate-bot
renovate-bot force-pushed the renovate/k8s.io-utils-digest branch from e3fbd0d to 228c137 Compare July 12, 2026 17:01
@sonarqubecloud

Copy link
Copy Markdown

@github-actions

Copy link
Copy Markdown

This pull request has been automatically marked as stale because it has not had recent activity. It will be closed in 30 days if no further activity occurs. Thank you for your contributions.

@github-actions github-actions Bot added the Stale label Sep 11, 2026
@renovate-bot
renovate-bot force-pushed the renovate/k8s.io-utils-digest branch from 228c137 to 31bb114 Compare September 15, 2026 15:21
Signed-off-by: Mend Renovate <bot@renovateapp.com>
@renovate-bot
renovate-bot force-pushed the renovate/k8s.io-utils-digest branch from 31bb114 to 63a7b76 Compare September 16, 2026 17:13
@sonarqubecloud

Copy link
Copy Markdown

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file size/XS

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants