Skip to content

Updating Base Account Quickstarts - #179

Merged
ericbrown99 merged 2 commits into
masterfrom
update-base-account-quickstart
Aug 7, 2025
Merged

ericbrown99 merged 2 commits into
masterfrom
update-base-account-quickstart

Conversation

@youssefea

Copy link
Copy Markdown
Contributor

What changed? Why?

Adding a new quickstart for HTML + JS simple implementation

* remove callouts

* update case

* remove base pay button from accept payments page

* remove siwb buttons

* update background buttons
@cb-heimdall

cb-heimdall commented Aug 7, 2025 •

Copy link
Copy Markdown
Collaborator

✅ Heimdall Review Status

Requirement Status More Info
Reviews ✅ 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@ericbrown99
ericbrown99 merged commit d99443a into master Aug 7, 2025
@ericbrown99
ericbrown99 deleted the update-base-account-quickstart branch August 7, 2025 20:59
madisoncarter1234 pushed a commit to madisoncarter1234/docs that referenced this pull request Oct 27, 2025

@octavio12345300 octavio12345300 left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

List

@cb-heimdall

Copy link
Copy Markdown
Collaborator

Review Error for octavio12345300 @ 2026-03-26 01:27:45 UTC
User failed mfa authentication, either user does not exist or public email is not set on your github profile. \ see go/mfa-help

This branch was successfully deployed

1 active deployment
staging - docs — fe69e318 Deployed Aug 7, 2025 by mintlify[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants