Skip to content

Add nodeSelector to override system deployments by operator - #3789

Merged
knative-prow-robot merged 6 commits into
knative:mkdocsfrom
nak3:add-nodelsector-to-operator-mkdoc
Jun 15, 2021
Merged

knative-prow-robot merged 6 commits into
knative:mkdocsfrom
nak3:add-nodelsector-to-operator-mkdoc

Conversation

@nak3

@nak3 nak3 commented Jun 14, 2021

Copy link
Copy Markdown
Contributor

This patch adds docs for knative/operator#658.
It adds nodeSelector to override system deployments by operator.

/cc @houshengbo

This patch adds docs for knative/operator#658.
It adds `nodeSelector` to override system deployments by operator.
@google-cla google-cla Bot added the cla: yes Indicates the PR's author has signed the CLA. label Jun 14, 2021
@knative-prow-robot knative-prow-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Jun 14, 2021
@netlify

netlify Bot commented Jun 14, 2021

Copy link
Copy Markdown

✔️ Deploy Preview for dev-knative ready!

🔨 Explore the source changes: eb5acca

🔍 Inspect the deploy log: https://app.netlify.com/sites/dev-knative/deploys/60c7f1c973bda200072c5cfb

😎 Browse the preview: https://deploy-preview-3789--dev-knative.netlify.app/development/admin/install/operator/configuring-serving-cr

Comment thread docs/admin/install/operator/configuring-serving-cr.md Outdated
Comment thread docs/admin/install/operator/configuring-serving-cr.md Outdated
Comment thread docs/admin/install/operator/configuring-serving-cr.md Outdated
Comment thread docs/admin/install/operator/configuring-serving-cr.md Outdated
nak3 and others added 4 commits June 15, 2021 09:17
Co-authored-by: Ashleigh Brennan <abrennan@redhat.com>
Co-authored-by: Ashleigh Brennan <abrennan@redhat.com>
Co-authored-by: Ashleigh Brennan <abrennan@redhat.com>
Co-authored-by: Ashleigh Brennan <abrennan@redhat.com>
@nak3

nak3 commented Jun 15, 2021

Copy link
Copy Markdown
Contributor Author

Thank you @abrennan89 ! Updated.

@csantanapr csantanapr left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

thanks
/lgtm
/aprove

@knative-prow-robot knative-prow-robot added the lgtm Indicates that a PR is ready to be merged. label Jun 15, 2021
@csantanapr

Copy link
Copy Markdown
Member

/approve

@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: csantanapr, nak3

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow-robot knative-prow-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 15, 2021
@knative-prow-robot
knative-prow-robot merged commit d4af310 into knative:mkdocs Jun 15, 2021
RichardJJG pushed a commit to RichardJJG/docs that referenced this pull request Jul 1, 2021
…e#3789)

* Add `nodeSelector` to override system deployments by operator

This patch adds docs for knative/operator#658.
It adds `nodeSelector` to override system deployments by operator.

* Fix broken format

* Update docs/admin/install/operator/configuring-serving-cr.md

Co-authored-by: Ashleigh Brennan <abrennan@redhat.com>

* Update docs/admin/install/operator/configuring-serving-cr.md

Co-authored-by: Ashleigh Brennan <abrennan@redhat.com>

* Update docs/admin/install/operator/configuring-serving-cr.md

Co-authored-by: Ashleigh Brennan <abrennan@redhat.com>

* Update docs/admin/install/operator/configuring-serving-cr.md

Co-authored-by: Ashleigh Brennan <abrennan@redhat.com>

Co-authored-by: Ashleigh Brennan <abrennan@redhat.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cla: yes Indicates the PR's author has signed the CLA. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants