Skip to content

[release-v1.17] Update golang x libraries to latest versions - #4169

Merged
Kaustubh-pande merged 1 commit into
openshift-knative:release-1.37from
dsimansk:pr/update-go-libs
Sep 16, 2026
Merged

Kaustubh-pande merged 1 commit into
openshift-knative:release-1.37from
dsimansk:pr/update-go-libs

Conversation

@dsimansk

Copy link
Copy Markdown
Contributor

Fixes JIRA #

Proposed Changes

  • [release-v1.17] Update golang x libraries to latest versions

/cc @Kaustubh-pande

@Kaustubh-pande

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci

openshift-ci Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk, Kaustubh-pande

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [Kaustubh-pande,dsimansk]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@Kaustubh-pande

Copy link
Copy Markdown
Contributor

/override ci/prow/420-kitchensink-upgrade ci/prow/421-kitchensink-e2e ci/prow/421-mesh-e2e ci/prow/421-mesh-upgrade ci/prow/421-operator-e2e ci/prow/421-test-upgrade ci/prow/421-upstream-e2e ci/prow/421-upstream-e2e-kafka

@openshift-ci

openshift-ci Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

@Kaustubh-pande: Overrode contexts on behalf of Kaustubh-pande: ci/prow/420-kitchensink-upgrade, ci/prow/421-kitchensink-e2e, ci/prow/421-mesh-e2e, ci/prow/421-mesh-upgrade, ci/prow/421-operator-e2e, ci/prow/421-test-upgrade, ci/prow/421-upstream-e2e, ci/prow/421-upstream-e2e-kafka

Details

In response to this:

/override ci/prow/420-kitchensink-upgrade ci/prow/421-kitchensink-e2e ci/prow/421-mesh-e2e ci/prow/421-mesh-upgrade ci/prow/421-operator-e2e ci/prow/421-test-upgrade ci/prow/421-upstream-e2e ci/prow/421-upstream-e2e-kafka

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci

openshift-ci Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

@dsimansk: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/421-test-upgrade d33d809 link true /test 421-test-upgrade
ci/prow/421-upstream-e2e-kafka d33d809 link false /test 421-upstream-e2e-kafka
ci/prow/421-kitchensink-e2e d33d809 link false /test 421-kitchensink-e2e
ci/prow/421-operator-e2e d33d809 link true /test 421-operator-e2e
ci/prow/421-mesh-e2e d33d809 link false /test 421-mesh-e2e
ci/prow/421-mesh-upgrade d33d809 link false /test 421-mesh-upgrade
ci/prow/421-upstream-e2e d33d809 link false /test 421-upstream-e2e

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@Kaustubh-pande
Kaustubh-pande merged commit 49a77c7 into openshift-knative:release-1.37 Sep 16, 2026
11 of 21 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants