Skip to content

[release-1.37] Run make generated-files - #4176

Merged
Kaustubh-pande merged 5 commits into
release-1.37from
auto/update-generated-files-release-1.37
Sep 18, 2026
Merged

Kaustubh-pande merged 5 commits into
release-1.37from
auto/update-generated-files-release-1.37

Conversation

@serverless-qe

Copy link
Copy Markdown
Contributor

Generated by running make generated-files

@Kaustubh-pande

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@openshift-ci

openshift-ci Bot commented Sep 18, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Kaustubh-pande, serverless-qe

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@Kaustubh-pande
Kaustubh-pande merged commit c88e596 into release-1.37 Sep 18, 2026
6 of 8 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants