feat: setup project with solidui-cli using deno - #181
Merged
Merged
Conversation
Using `deno run @solidui-cli@latest init`. Signed-off-by: João Augusto Costa Branco Marado Torres <torres.dev@disroot.org>
|
The latest updates on your projects. Learn more about Vercel for Git ↗︎
|
🦋 Changeset detectedLatest commit: 93e8bbd The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Using `deno run npm:solidui-cli@latest add button` for example. Signed-off-by: João Augusto Costa Branco Marado Torres <torres.dev@disroot.org>
Contributor
Author
|
Turns out adding components with the CLI has the same problem, but same solution. |
Signed-off-by: João Augusto Costa Branco Marado Torres <torres.dev@disroot.org>
Signed-off-by: João Augusto Costa Branco Marado Torres <torres.dev@disroot.org>
Owner
|
Thanks for your hard work! :) |
GitTuanKiet
added a commit
to GitTuanKiet/solid-ui
that referenced
this pull request
Jan 21, 2026
Error caused by stefan-karger#181. Run bunx solidui-cli@latest add button, error: ExecaError: Command failed with exit code 1: bun add '' '@kobalte/core'
This branch had an error being deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Using
deno run npm:solidui-cli@latest init.When doing it before it would fail because it needs the
npm:prefix on each dependency which it didn't.