Skip to content

chore: version packages - #286

Merged
thymikee merged 2 commits into
mainfrom
changeset-release/main
Apr 28, 2025
Merged

thymikee merged 2 commits into
mainfrom
changeset-release/main

Conversation

@github-actions

@github-actions github-actions Bot commented Apr 25, 2025 •

Copy link
Copy Markdown
Contributor

This PR was opened by the Changesets release GitHub action. When you're ready to do a release, you can merge this and the packages will be published to npm automatically. If you're not ready to do a release yet, that's fine, whenever you add more changesets to main, this PR will be updated.

Releases

@rnef/cli@0.7.3

Patch Changes

  • Updated dependencies [0be7484]
    • @rnef/config@0.7.3
    • @rnef/tools@0.7.3

@rnef/config@0.7.3

Patch Changes

  • 0be7484: feat: refactor and expose remote cache provider
  • Updated dependencies [0be7484]
    • @rnef/tools@0.7.3

@rnef/create-app@0.7.3

Patch Changes

  • d06dbe9: chore: add install deps to create-app flow
  • Updated dependencies [0be7484]
    • @rnef/tools@0.7.3

@rnef/platform-android@0.7.3

Patch Changes

  • 0be7484: feat: refactor and expose remote cache provider
  • b410121: feat: structured build settings information
  • Updated dependencies [0be7484]
    • @rnef/tools@0.7.3

@rnef/platform-apple-helpers@0.7.3

Patch Changes

  • f11fd91: fix: always use udid from matched device
  • 0be7484: feat: refactor and expose remote cache provider
  • b410121: feat: structured build settings information
  • ab730d9: chore: improve error messages around installing pods; run codegen
  • Updated dependencies [0be7484]
    • @rnef/tools@0.7.3

@rnef/platform-ios@0.7.3

Patch Changes

  • ab730d9: chore: improve error messages around installing pods; run codegen
  • Updated dependencies [f11fd91]
  • Updated dependencies [0be7484]
  • Updated dependencies [b410121]
  • Updated dependencies [ab730d9]
    • @rnef/platform-apple-helpers@0.7.3
    • @rnef/tools@0.7.3

@rnef/plugin-brownfield-android@0.7.3

Patch Changes

  • Updated dependencies [0be7484]
  • Updated dependencies [b410121]
    • @rnef/platform-android@0.7.3
    • @rnef/tools@0.7.3

@rnef/plugin-brownfield-ios@0.7.3

Patch Changes

  • ab730d9: chore: improve error messages around installing pods; run codegen
  • Updated dependencies [f11fd91]
  • Updated dependencies [0be7484]
  • Updated dependencies [b410121]
  • Updated dependencies [ab730d9]
    • @rnef/platform-apple-helpers@0.7.3
    • @rnef/tools@0.7.3

@rnef/plugin-metro@0.7.3

Patch Changes

  • 86522ec: feat: move start from @react-native/community-cli-plugin
  • Updated dependencies [0be7484]
    • @rnef/tools@0.7.3

@rnef/plugin-repack@0.7.3

Patch Changes

  • Updated dependencies [0be7484]
    • @rnef/tools@0.7.3

@rnef/tools@0.7.3

Patch Changes

  • 0be7484: feat: refactor and expose remote cache provider

@rnef/test-helpers@0.7.3

@vercel

vercel Bot commented Apr 25, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
rnef ⬜️ Ignored (Inspect) Visit Preview Apr 28, 2025 1:44pm

@github-actions
github-actions Bot force-pushed the changeset-release/main branch 4 times, most recently from a6b83e5 to a69a123 Compare April 28, 2025 13:43
@github-actions
github-actions Bot force-pushed the changeset-release/main branch from a69a123 to 91b8853 Compare April 28, 2025 13:43
@thymikee
thymikee merged commit f3a32c8 into main Apr 28, 2025
@thymikee
thymikee deleted the changeset-release/main branch April 28, 2025 13:48
thymikee added a commit that referenced this pull request Apr 28, 2025
* origin/main:
  chore: v0.7.4 (#293)
  chore: install deps after changeset version (#294)
  chore: improve version script to change package.json version for template
  fix: run codegen script in sourceDir (#292)
  chore: update website address
  chore: version packages (#286)
  feat: structured build settings information (#291)
  chore: add install deps to create-app flow (#290)
  chore: improve error messages around installing pods; run codegen (#289)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant