Skip to content

chore: add install deps to create-app flow - #290

Merged
thymikee merged 3 commits into
mainfrom
chore/improve-intro
Apr 28, 2025
Merged

thymikee merged 3 commits into
mainfrom
chore/improve-intro

Conversation

@thymikee

Copy link
Copy Markdown
Member

Summary


┌  Welcome to React Native Enterprise Framework!
│
◇  What is your app named?
│  asdf
│
◇  What platforms do you want to start with?
│  ios, android
│
◇  Which bundler do you want to use?
│  metro
│
◇  Which remote cache provider do you want to use?
│  GitHub Actions
│
◇  Do you want to install dependencies?
│  Yes
│
◇  Applied template, platforms and plugins
│
◇  Installed dependencies with npm
│
◇  Git repo initialized
│
◇  Next steps ─────────────────────────────────────╮
│                                                  │
│  cd asdf                                         │
│  npm run start      starts dev server            │
│  npm run ios        builds and runs iOS app      │
│  npm run android    builds and runs Android app  │
│                                                  │
├──────────────────────────────────────────────────╯
│
└  Success 🎉.

Test plan

@vercel

vercel Bot commented Apr 28, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
rnef ⬜️ Ignored (Inspect) Visit Preview Apr 28, 2025 1:58am

@thymikee
thymikee merged commit d06dbe9 into main Apr 28, 2025
@thymikee
thymikee deleted the chore/improve-intro branch April 28, 2025 13:42
thymikee added a commit that referenced this pull request Apr 28, 2025
* origin/main:
  chore: v0.7.4 (#293)
  chore: install deps after changeset version (#294)
  chore: improve version script to change package.json version for template
  fix: run codegen script in sourceDir (#292)
  chore: update website address
  chore: version packages (#286)
  feat: structured build settings information (#291)
  chore: add install deps to create-app flow (#290)
  chore: improve error messages around installing pods; run codegen (#289)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant