Skip to content

chore: v0.7.4 - #293

Merged
thymikee merged 1 commit into
mainfrom
changeset-release/main
Apr 28, 2025
Merged

thymikee merged 1 commit into
mainfrom
changeset-release/main

Conversation

@github-actions

Copy link
Copy Markdown
Contributor

This PR was opened by the Changesets release GitHub action. When you're ready to do a release, you can merge this and the packages will be published to npm automatically. If you're not ready to do a release yet, that's fine, whenever you add more changesets to main, this PR will be updated.

Releases

@rnef/cli@0.7.4

Patch Changes

  • @rnef/config@0.7.4
  • @rnef/tools@0.7.4

@rnef/config@0.7.4

Patch Changes

  • @rnef/tools@0.7.4

@rnef/create-app@0.7.4

Patch Changes

  • @rnef/tools@0.7.4

@rnef/platform-android@0.7.4

Patch Changes

  • @rnef/tools@0.7.4

@rnef/platform-apple-helpers@0.7.4

Patch Changes

  • 44ebc93: fix: run codegen script in sourceDir
    • @rnef/tools@0.7.4

@rnef/platform-ios@0.7.4

Patch Changes

  • Updated dependencies [44ebc93]
    • @rnef/platform-apple-helpers@0.7.4
    • @rnef/tools@0.7.4

@rnef/plugin-brownfield-android@0.7.4

Patch Changes

  • @rnef/platform-android@0.7.4
  • @rnef/tools@0.7.4

@rnef/plugin-brownfield-ios@0.7.4

Patch Changes

  • Updated dependencies [44ebc93]
    • @rnef/platform-apple-helpers@0.7.4
    • @rnef/tools@0.7.4

@rnef/plugin-metro@0.7.4

Patch Changes

  • @rnef/tools@0.7.4

@rnef/plugin-repack@0.7.4

Patch Changes

  • @rnef/tools@0.7.4

@rnef/tools@0.7.4

@rnef/test-helpers@0.7.4

@vercel

vercel Bot commented Apr 28, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
rnef ⬜️ Ignored (Inspect) Visit Preview Apr 28, 2025 5:49pm

@github-actions
github-actions Bot force-pushed the changeset-release/main branch from ec2f8c8 to 331eb51 Compare April 28, 2025 17:41
@thymikee thymikee changed the title chore: version packages chore: v0.7.4 Apr 28, 2025
@github-actions github-actions Bot changed the title chore: v0.7.4 chore: version packages Apr 28, 2025
@github-actions
github-actions Bot force-pushed the changeset-release/main branch from 331eb51 to 07976cd Compare April 28, 2025 17:48
@thymikee thymikee changed the title chore: version packages chore: v0.7.4 Apr 28, 2025
@thymikee
thymikee merged commit 555f27a into main Apr 28, 2025
@thymikee
thymikee deleted the changeset-release/main branch April 28, 2025 17:50
thymikee added a commit that referenced this pull request Apr 28, 2025
* origin/main:
  chore: v0.7.4 (#293)
  chore: install deps after changeset version (#294)
  chore: improve version script to change package.json version for template
  fix: run codegen script in sourceDir (#292)
  chore: update website address
  chore: version packages (#286)
  feat: structured build settings information (#291)
  chore: add install deps to create-app flow (#290)
  chore: improve error messages around installing pods; run codegen (#289)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant