Skip to content

feat: add remote-cache command - #263

Merged
thymikee merged 45 commits into
mainfrom
feat/remote-cache-plugin
May 17, 2025
Merged

thymikee merged 45 commits into
mainfrom
feat/remote-cache-plugin

Conversation

@thymikee

@thymikee thymikee commented Apr 15, 2025

Copy link
Copy Markdown
Member

Summary

Add remote-cache command with CRUD for artifacts management based on remoteCacheProvider. Available subcommands:

  • list
  • download
  • upload
  • delete

Test plan

@vercel

vercel Bot commented Apr 15, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
rnef ⬜️ Ignored (Inspect) Visit Preview May 17, 2025 11:35am

@thymikee
thymikee force-pushed the feat/remote-cache-plugin branch 2 times, most recently from b96f52e to 57dbb48 Compare April 15, 2025 19:39
@thymikee
thymikee force-pushed the refactor/remote-cache branch from df80f35 to a9d1999 Compare April 18, 2025 14:57
@thymikee
thymikee force-pushed the refactor/remote-cache branch 2 times, most recently from a4c4390 to a96287b Compare April 28, 2025 02:07
Base automatically changed from refactor/remote-cache to main April 28, 2025 02:57
* origin/main:
  feat: refactor and expose remote cache provider (#260)
  fix: move start from @react-native/community-cli-plugin (#288)
  fix: update gradle-wrapper (#287)
  fix: always use udid from matched device (#281)
  chore: v0.7.2 (#279)
  docs: Update migrating-from-community-cli.mdx (#284)
  docs: attribute @expo/fingerprint (#285)
  fix: support plugin in create-app
  chore: v0.7.1 (#278)
  fix: autolinking config missing project (#277)
  fix: bump template to v0.7.0
  chore: cleanup ios brownfield android mention
  chore: v0.7.0 (#274)
  feat: upgrade templates to 0.79 (#275)
  fix: incorrectly resolving root (#276)
  chore: simplify welcome message and improve prompts (#273)
  docs: add logo (#272)
  chore: v0.6.2 (#268)
  fix: resolve project root to absolute path (#269)
* origin/main:
  chore: v0.7.4 (#293)
  chore: install deps after changeset version (#294)
  chore: improve version script to change package.json version for template
  fix: run codegen script in sourceDir (#292)
  chore: update website address
  chore: version packages (#286)
  feat: structured build settings information (#291)
  chore: add install deps to create-app flow (#290)
  chore: improve error messages around installing pods; run codegen (#289)
thymikee added 13 commits May 16, 2025 19:57
* feat/remote-cache-rework:
  rework provider plugin to be configurable
  make download return Reponse and move logic to RNEF
  get rid of local artifact path as it's handled by RNEF; pass targetURL
  refactor getLocalBinaryPath
  simplify upload return
  remove id
  simplify return to array only
  changeset
  mark delete and upload as optional
  unify interface
  add test
  feat: add missing methods to remote cache provider
  chore: add Sitemap plugin to docs (#337)
  docs: support `llms.txt` (#335)
  chore: upgrade to rspress v2 (#334)
@thymikee
thymikee changed the base branch from main to feat/remote-cache-rework May 16, 2025 19:47
@thymikee thymikee changed the title [WIP] feat: add remote-cache command and internal plugin feat: add remote-cache command and internal plugin May 16, 2025
@thymikee thymikee changed the title feat: add remote-cache command and internal plugin feat: add remote-cache command May 16, 2025
Base automatically changed from feat/remote-cache-rework to main May 17, 2025 10:25

@szymonrybczak szymonrybczak left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM, very helpful.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants